Skip to content

[release-4.17] denylist: extend snooze date for ext.config.shared.rpm-ostree.kernel-replace - #1969

Merged
sdodson merged 1 commit into
openshift:release-4.17from
openshift-cherrypick-robot:cherry-pick-1968-to-release-4.17
Sep 15, 2026
Merged

sdodson merged 1 commit into
openshift:release-4.17from
openshift-cherrypick-robot:cherry-pick-1968-to-release-4.17

Conversation

@openshift-cherrypick-robot

Copy link
Copy Markdown

This is an automated cherry-pick of #1968

/assign sdodson

…replace

The test is still failing and is blocking CI. Snoozing it until 2026-09-21

ref: coreos/rhel-coreos-config#305
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Pipeline controller notification
This repo is configured to use the pipeline controller. Second-stage tests will be triggered either automatically or after lgtm label is added, depending on the repository configuration. The pipeline controller will automatically detect which contexts are required and will utilize /test Prow commands to trigger the second stage.

For optional jobs, comment /test ? to see a list of all defined jobs. To trigger manually all jobs from second stage use /pipeline required command.

This repository is configured in: LGTM mode

@openshift-ci
openshift-ci Bot requested review from c4rt0 and ravanelli September 15, 2026 22:33
@sdodson sdodson added lgtm Indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. labels Sep 15, 2026
@sdodson
sdodson merged commit a61d06c into openshift:release-4.17 Sep 15, 2026
0 of 4 checks passed
@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: openshift-cherrypick-robot

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants