Skip to content

denylist: extend snooze date for ext.config.shared.rpm-ostree.kernel-replace - #1968

Merged
sdodson merged 1 commit into
openshift:release-4.18from
Roshan-R:patch-1
Sep 15, 2026
Merged

sdodson merged 1 commit into
openshift:release-4.18from
Roshan-R:patch-1

Conversation

@Roshan-R

Copy link
Copy Markdown
Contributor

The test is still failing and is blocking CI. Snoozing it until 2026-09-21

ref: coreos/rhel-coreos-config#305

…replace

The test is still failing and is blocking CI. Snoozing it until 2026-09-21

ref: coreos/rhel-coreos-config#305
@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown
Contributor

Pipeline controller notification
This repo is configured to use the pipeline controller. Second-stage tests will be triggered either automatically or after lgtm label is added, depending on the repository configuration. The pipeline controller will automatically detect which contexts are required and will utilize /test Prow commands to trigger the second stage.

For optional jobs, comment /test ? to see a list of all defined jobs. To trigger manually all jobs from second stage use /pipeline required command.

This repository is configured in: LGTM mode

@openshift-ci
openshift-ci Bot requested review from jschintag and marmijo September 15, 2026 09:19
@jbtrystram

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 15, 2026
@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jbtrystram, Roshan-R

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 15, 2026
@Roshan-R

Copy link
Copy Markdown
Contributor Author

/retest

@sdodson sdodson added the backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. label Sep 15, 2026
@sdodson

sdodson commented Sep 15, 2026

Copy link
Copy Markdown
Member

/test rhcos-9-build-test-qemu

@sdodson sdodson added the jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. label Sep 15, 2026
@sdodson

sdodson commented Sep 15, 2026

Copy link
Copy Markdown
Member

/verified by CI

@openshift-ci-robot openshift-ci-robot added the verified Signifies that the PR passed pre-merge verification criteria label Sep 15, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@sdodson: This PR has been marked as verified by CI.

Details

In response to this:

/verified by CI

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@sdodson
sdodson merged commit b5763b8 into openshift:release-4.18 Sep 15, 2026
3 of 5 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. lgtm Indicates that a PR is ready to be merged. verified Signifies that the PR passed pre-merge verification criteria

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants