chore improvements on create-granite-app - #316
Open
mym0404 wants to merge 5 commits into
Open
Conversation
mym0404
requested review from
gronxb,
heecheolman,
intmain,
jingjing2222,
l2hyunwoo,
leegeunhyeok and
raon0211
as code owners
June 17, 2026 14:44
🦋 Changeset detectedLatest commit: 9359f49 The changes in this PR will be included in the next version bump. This PR includes changesets to release 28 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
|
@mym0404 is attempting to deploy a commit to the Toss Team on Vercel. A member of the Team first needs to authorize it. |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Overview
blocked by #316
I encountered these warnings from a generated
shared bundle appduring easygoing browsing.While I was working on these fixes, I got pulled into a couple of small cleanup tasks and ended up fixing those as well.
Changes
1.
path.join->path.resolveCLI can accept an absolute path now.
2. Add
typeimport keyword for auto-generated route import statements.plugin-routeraffected3. Remove redundant
import Reactfrom the template.Evidences (use
automaticjsx transformation option)granite/packages/mpack/src/bundler/Bundler.ts
Line 139 in 94896a3
granite/packages/mpack/src/bundler/plugins/transformPlugin/steps/createFullyTransformStep.ts
Line 52 in 94896a3
granite/packages/mpack/src/bundler/plugins/transformPlugin/steps/createTransformToHermesSyntaxStep.ts
Line 41 in 94896a3
4. Use
selectovermultiselectwhen choosing linter.5.
lintyarn command ofbiometemplate doesn't fix code anymoreeslint(check only).biome check --writehad been silently fixing warnings, the test had been passed for a while.Test Plan
yarn workspace create-granite-app test:no-parallel