Tidy up redundant yarn dependencies constraints check - #314
Open
mym0404 wants to merge 3 commits into
Open
Conversation
mym0404
requested review from
gronxb,
heecheolman,
intmain,
jingjing2222,
l2hyunwoo,
leegeunhyeok and
raon0211
as code owners
June 17, 2026 01:32
|
@mym0404 is attempting to deploy a commit to the Toss Team on Vercel. A member of the Team first needs to authorize it. |
🦋 Changeset detectedLatest commit: b391ddc The changes in this PR will be included in the next version bump. This PR includes changesets to release 0 packagesWhen changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Overview
git logsaysyarn.config.cjsis renamed to.mjsat one point(#64).yarndoesn't seem to support this. docs.I checked in my local, and it didn't make sense anymore.
If it hasn't been doing nothing for a while now, and it's been okay, I thought maybe we could try and remove it? From my perspective, this project already has a strict deps version management rules by employing
yarn catalog.